Skip to content
This repository was archived by the owner on Mar 24, 2026. It is now read-only.

Images: Bump GCB Docker GCloud to v20251110-7ccd542560. - #14155

Merged
k8s-ci-robot merged 1 commit into
kubernetes:mainfrom
Gacko:sjitz
Nov 11, 2025
Merged

k8s-ci-robot merged 1 commit into
kubernetes:mainfrom
Gacko:sjitz

Conversation

@Gacko

@Gacko Gacko commented Nov 10, 2025

Copy link
Copy Markdown
Member

/triage accepted
/kind cleanup
/priority backlog
/cc @cpanato @strongjz @tao12345666333
/hold

@k8s-ci-robot
k8s-ci-robot requested a review from cpanato November 10, 2025 20:04
@k8s-ci-robot k8s-ci-robot added the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Nov 10, 2025
@k8s-ci-robot k8s-ci-robot added triage/accepted Indicates an issue or PR is ready to be actively worked on. kind/cleanup Categorizes issue or PR as related to cleaning up code, process, or technical debt. priority/backlog Higher priority than priority/awaiting-more-evidence. labels Nov 10, 2025
@netlify

netlify Bot commented Nov 10, 2025

Copy link
Copy Markdown

Deploy Preview for kubernetes-ingress-nginx canceled.

Name Link
🔨 Latest commit f5f5f55
🔍 Latest deploy log https://app.netlify.com/projects/kubernetes-ingress-nginx/deploys/6912453a4b60230007e2ab0f

@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. area/helm Issues or PRs related to helm charts approved Indicates a PR has been approved by an approver from all required OWNERS files. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Nov 10, 2025

@cpanato cpanato left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Nov 11, 2025
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: cpanato, Gacko

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@Gacko

Gacko commented Nov 11, 2025

Copy link
Copy Markdown
Member Author

/unhold

@k8s-ci-robot k8s-ci-robot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Nov 11, 2025
@k8s-ci-robot
k8s-ci-robot merged commit 1f4e7a3 into kubernetes:main Nov 11, 2025
28 checks passed
@Gacko
Gacko deleted the sjitz branch November 11, 2025 19:38
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. area/helm Issues or PRs related to helm charts cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. kind/cleanup Categorizes issue or PR as related to cleaning up code, process, or technical debt. lgtm "Looks good to me", indicates that a PR is ready to be merged. priority/backlog Higher priority than priority/awaiting-more-evidence. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. triage/accepted Indicates an issue or PR is ready to be actively worked on.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants