Skip to content
This repository was archived by the owner on Mar 24, 2026. It is now read-only.

Images: Trigger other builds (2/2). - #14231

Merged
Gacko merged 1 commit into
kubernetes:mainfrom
Gacko:qswyt
Nov 27, 2025
Merged

Gacko merged 1 commit into
kubernetes:mainfrom
Gacko:qswyt

Conversation

@Gacko

@Gacko Gacko commented Nov 27, 2025

Copy link
Copy Markdown
Member

/triage accepted
/kind cleanup
/priority backlog
/cc @cpanato @strongjz @tao12345666333
/hold

@k8s-ci-robot k8s-ci-robot added do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. triage/accepted Indicates an issue or PR is ready to be actively worked on. kind/cleanup Categorizes issue or PR as related to cleaning up code, process, or technical debt. labels Nov 27, 2025
@netlify

netlify Bot commented Nov 27, 2025

Copy link
Copy Markdown

Deploy Preview for kubernetes-ingress-nginx canceled.

Name Link
🔨 Latest commit 9fc3667
🔍 Latest deploy log https://app.netlify.com/projects/kubernetes-ingress-nginx/deploys/6928244d55cd1300089c77a2

@k8s-ci-robot k8s-ci-robot added the priority/backlog Higher priority than priority/awaiting-more-evidence. label Nov 27, 2025
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: Gacko

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added approved Indicates a PR has been approved by an approver from all required OWNERS files. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Nov 27, 2025
@Gacko

Gacko commented Nov 27, 2025

Copy link
Copy Markdown
Member Author

/unhold

@k8s-ci-robot k8s-ci-robot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Nov 27, 2025
@Gacko
Gacko merged commit cc7f992 into kubernetes:main Nov 27, 2025
26 of 27 checks passed
@Gacko
Gacko deleted the qswyt branch November 27, 2025 12:45
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. kind/cleanup Categorizes issue or PR as related to cleaning up code, process, or technical debt. priority/backlog Higher priority than priority/awaiting-more-evidence. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. triage/accepted Indicates an issue or PR is ready to be actively worked on.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants