Skip to content
This repository was archived by the owner on Mar 24, 2026. It is now read-only.

CI: Update Kubernetes to v1.34.3. - #14281

Merged
k8s-ci-robot merged 1 commit into
kubernetes:mainfrom
Gacko:iirvu
Dec 10, 2025
Merged

k8s-ci-robot merged 1 commit into
kubernetes:mainfrom
Gacko:iirvu

Conversation

@Gacko

@Gacko Gacko commented Dec 10, 2025

Copy link
Copy Markdown
Member

/triage accepted
/kind cleanup
/priority backlog
/cc @cpanato @strongjz @tao12345666333

@k8s-ci-robot k8s-ci-robot added triage/accepted Indicates an issue or PR is ready to be actively worked on. kind/cleanup Categorizes issue or PR as related to cleaning up code, process, or technical debt. priority/backlog Higher priority than priority/awaiting-more-evidence. labels Dec 10, 2025
@netlify

netlify Bot commented Dec 10, 2025

Copy link
Copy Markdown

Deploy Preview for kubernetes-ingress-nginx canceled.

Name Link
🔨 Latest commit 113d2c9
🔍 Latest deploy log https://app.netlify.com/projects/kubernetes-ingress-nginx/deploys/693931afa1d29d0008124025

@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Dec 10, 2025
@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Dec 10, 2025
@k8s-ci-robot k8s-ci-robot added the size/M Denotes a PR that changes 30-99 lines, ignoring generated files. label Dec 10, 2025

@cpanato cpanato left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm
/approve

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Dec 10, 2025
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: cpanato, Gacko

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot
k8s-ci-robot merged commit 59c0bbe into kubernetes:main Dec 10, 2025
40 checks passed
@Gacko
Gacko deleted the iirvu branch December 10, 2025 09:48
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. kind/cleanup Categorizes issue or PR as related to cleaning up code, process, or technical debt. lgtm "Looks good to me", indicates that a PR is ready to be merged. priority/backlog Higher priority than priority/awaiting-more-evidence. size/M Denotes a PR that changes 30-99 lines, ignoring generated files. triage/accepted Indicates an issue or PR is ready to be actively worked on.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants