Skip to content
This repository was archived by the owner on Mar 24, 2026. It is now read-only.

Images: Trigger other builds (1/2). - #14422

Merged
Gacko merged 1 commit into
kubernetes:mainfrom
Gacko:khsqo
Jan 21, 2026
Merged

Gacko merged 1 commit into
kubernetes:mainfrom
Gacko:khsqo

Conversation

@Gacko

@Gacko Gacko commented Jan 21, 2026

Copy link
Copy Markdown
Member

/triage accepted
/kind cleanup
/priority backlog
/cc @cpanato @strongjz @tao12345666333
/hold

@k8s-ci-robot k8s-ci-robot added do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. triage/accepted Indicates an issue or PR is ready to be actively worked on. kind/cleanup Categorizes issue or PR as related to cleaning up code, process, or technical debt. priority/backlog Higher priority than priority/awaiting-more-evidence. labels Jan 21, 2026
@netlify

netlify Bot commented Jan 21, 2026

Copy link
Copy Markdown

Deploy Preview for kubernetes-ingress-nginx canceled.

Name Link
🔨 Latest commit 19aa7f8
🔍 Latest deploy log https://app.netlify.com/projects/kubernetes-ingress-nginx/deploys/697075e7557e6b0008206621

@k8s-ci-robot k8s-ci-robot added the cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. label Jan 21, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: Gacko

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added approved Indicates a PR has been approved by an approver from all required OWNERS files. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Jan 21, 2026
@Gacko

Gacko commented Jan 21, 2026

Copy link
Copy Markdown
Member Author

/unhold

@k8s-ci-robot k8s-ci-robot removed the do-not-merge/hold Indicates that a PR should not merge because someone has issued a /hold command. label Jan 21, 2026
@Gacko Gacko added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jan 21, 2026
@Gacko
Gacko merged commit c26980a into kubernetes:main Jan 21, 2026
24 of 25 checks passed
@Gacko
Gacko deleted the khsqo branch January 21, 2026 09:17
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. kind/cleanup Categorizes issue or PR as related to cleaning up code, process, or technical debt. lgtm "Looks good to me", indicates that a PR is ready to be merged. priority/backlog Higher priority than priority/awaiting-more-evidence. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. triage/accepted Indicates an issue or PR is ready to be actively worked on.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants