Skip to content
This repository was archived by the owner on Mar 24, 2026. It is now read-only.

CI: Update Kubernetes to v1.35.3. - #14702

Merged
Gacko merged 1 commit into
kubernetes:release-1.14from
Gacko:gszqp
Mar 19, 2026
Merged

Gacko merged 1 commit into
kubernetes:release-1.14from
Gacko:gszqp

Conversation

@Gacko

@Gacko Gacko commented Mar 19, 2026

Copy link
Copy Markdown
Member

/triage accepted
/kind cleanup
/priority backlog
/cc @cpanato @strongjz @tao12345666333

@k8s-ci-robot k8s-ci-robot added triage/accepted Indicates an issue or PR is ready to be actively worked on. kind/cleanup Categorizes issue or PR as related to cleaning up code, process, or technical debt. priority/backlog Higher priority than priority/awaiting-more-evidence. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. labels Mar 19, 2026
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: Gacko

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added approved Indicates a PR has been approved by an approver from all required OWNERS files. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Mar 19, 2026
@Gacko
Gacko merged commit 1b0c278 into kubernetes:release-1.14 Mar 19, 2026
18 checks passed
@Gacko
Gacko deleted the gszqp branch March 19, 2026 11:52
swastik959 pushed a commit to swastik959/ingress-nginx that referenced this pull request Mar 23, 2026
Sign up for free to subscribe to this conversation on GitHub. Already have an account? Sign in.

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. kind/cleanup Categorizes issue or PR as related to cleaning up code, process, or technical debt. priority/backlog Higher priority than priority/awaiting-more-evidence. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. triage/accepted Indicates an issue or PR is ready to be actively worked on.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants