Skip to content

fix(remix-dev): update type for ServerBuild virtual module - #3743

Merged
mcansh merged 2 commits into
devfrom
logan/update-server-build-type
Jul 14, 2022
Merged

mcansh merged 2 commits into
devfrom
logan/update-server-build-type

Conversation

@mcansh

@mcansh mcansh commented Jul 14, 2022 •

Copy link
Copy Markdown
Contributor

closes remix-run/grunge-stack#85
closes remix-run/grunge-stack#89

x-ref #3349

Signed-off-by: Logan McAnsh logan@mcan.sh

Closes: #

  • Docs
  • Tests

Testing Strategy:

x-ref #3349

Signed-off-by: Logan McAnsh <logan@mcan.sh>
@changeset-bot

changeset-bot Bot commented Jul 14, 2022 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 6c6c6ea

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 16 packages
Name Type
remix Patch
@remix-run/dev Patch
create-remix Patch
@remix-run/eslint-config Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
@remix-run/cloudflare Patch
@remix-run/node Patch
@remix-run/deno Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/express Patch
@remix-run/netlify Patch
@remix-run/vercel Patch
@remix-run/architect Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@mcansh mcansh changed the title fix(remix-dev): update ServerBuild type fix(remix-dev): update type for ServerBuild virtual module Jul 14, 2022
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants