Skip to content

add serverBuildTarget deprecation changeset - #5354

Merged
jacob-ebey merged 1 commit into
devfrom
serverbuildtargetchangeset
Feb 2, 2023
Merged

jacob-ebey merged 1 commit into
devfrom
serverbuildtargetchangeset

Conversation

@jacob-ebey

Copy link
Copy Markdown
Member

Closes: #

  • Docs
  • Tests

Testing Strategy:

@changeset-bot

changeset-bot Bot commented Feb 2, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 2b0dcd9

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 18 packages
Name Type
@remix-run/dev Minor
create-remix Minor
@remix-run/css-bundle Minor
remix Minor
@remix-run/architect Minor
@remix-run/cloudflare Minor
@remix-run/cloudflare-pages Minor
@remix-run/cloudflare-workers Minor
@remix-run/deno Minor
@remix-run/eslint-config Minor
@remix-run/express Minor
@remix-run/netlify Minor
@remix-run/node Minor
@remix-run/react Minor
@remix-run/serve Minor
@remix-run/server-runtime Minor
@remix-run/testing Minor
@remix-run/vercel Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@jacob-ebey
jacob-ebey merged commit 79859c9 into dev Feb 2, 2023
@jacob-ebey
jacob-ebey deleted the serverbuildtargetchangeset branch February 2, 2023 23:08
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant