Skip to content

chore: sync FutureConfig interface - #5398

Merged
mcansh merged 1 commit into
devfrom
logan/sync-futureconfig-type
Feb 15, 2023
Merged

mcansh merged 1 commit into
devfrom
logan/sync-futureconfig-type

Conversation

@mcansh

@mcansh mcansh commented Feb 8, 2023

Copy link
Copy Markdown
Contributor

Signed-off-by: Logan McAnsh logan@mcan.sh
(cherry picked from commit 2d28768b17a0767c2723625c5acec51a02ec0d22)

Closes: #

  • Docs
  • Tests

Testing Strategy:

@changeset-bot

changeset-bot Bot commented Feb 8, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 2c42531

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 18 packages
Name Type
remix Patch
@remix-run/dev Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
create-remix Patch
@remix-run/css-bundle Patch
@remix-run/testing Patch
@remix-run/cloudflare Patch
@remix-run/deno Patch
@remix-run/node Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/architect Patch
@remix-run/express Patch
@remix-run/netlify Patch
@remix-run/vercel Patch
@remix-run/eslint-config Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@mcansh mcansh changed the title chore: sync FutureConfig interface chore: sync FutureConfig interface Feb 8, 2023
@mcansh
mcansh force-pushed the logan/sync-futureconfig-type branch from ea0f0a7 to a504966 Compare February 15, 2023 17:49
@mcansh mcansh mentioned this pull request Feb 15, 2023
@mcansh mcansh linked an issue Feb 15, 2023 that may be closed by this pull request
Comment thread packages/remix-server-runtime/server.ts Outdated
Comment thread packages/remix-dev/config.ts Outdated
Signed-off-by: Logan McAnsh <logan@mcan.sh>
@mcansh
mcansh force-pushed the logan/sync-futureconfig-type branch from a504966 to 2c42531 Compare February 15, 2023 18:39
@mcansh
mcansh merged commit 2f1e3f4 into dev Feb 15, 2023
@mcansh
mcansh deleted the logan/sync-futureconfig-type branch February 15, 2023 19:04
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-db97131-20230216 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

sync FutureConfig types

3 participants