Skip to content

Add support for mjs and cjs extensions with CSS imports - #5564

Merged
markdalgleish merged 7 commits into
remix-run:devfrom
devongovett:css-mjs
Mar 14, 2023
Merged

markdalgleish merged 7 commits into
remix-run:devfrom
devongovett:css-mjs

Conversation

@devongovett

@devongovett devongovett commented Feb 24, 2023 •

Copy link
Copy Markdown
Contributor

Closes: #5562

  • Docs
  • Tests

In the latest release of React Spectrum we added support for Node ESM, and now distribute a .mjs build. This appears to have broken the CSS loader with Remix: adobe/react-spectrum#4118. Looks like the .mjs and .cjs extensions were missing in the esbuild plugin filter, so I've gone ahead and added them here.

Testing Strategy: Updated integration tests, and tested by patching an example app with React Spectrum.

@changeset-bot

changeset-bot Bot commented Feb 24, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ad1fede

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 18 packages
Name Type
@remix-run/dev Patch
create-remix Patch
@remix-run/css-bundle Patch
remix Patch
@remix-run/architect Patch
@remix-run/cloudflare Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/deno Patch
@remix-run/eslint-config Patch
@remix-run/express Patch
@remix-run/netlify Patch
@remix-run/node Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
@remix-run/testing Patch
@remix-run/vercel Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@remix-cla-bot

remix-cla-bot Bot commented Feb 24, 2023

Copy link
Copy Markdown
Contributor

Hi @devongovett,

Welcome, and thank you for contributing to Remix!

Before we consider your pull request, we ask that you sign our Contributor License Agreement (CLA). We require this only once.

You may review the CLA and sign it by adding your name to contributors.yml.

Once the CLA is signed, the CLA Signed label will be added to the pull request.

If you have already signed the CLA and received this response in error, or if you have any questions, please contact us at hello@remix.run.

Thanks!

- The Remix team

@remix-cla-bot

remix-cla-bot Bot commented Feb 24, 2023

Copy link
Copy Markdown
Contributor

Thank you for signing the Contributor License Agreement. Let's get this merged! 🥳

@machour
machour requested a review from markdalgleish March 1, 2023 16:34
@markdalgleish
markdalgleish changed the base branch from main to dev March 13, 2023 23:26
@markdalgleish
markdalgleish merged commit 5782a43 into remix-run:dev Mar 14, 2023
@markdalgleish

Copy link
Copy Markdown
Member

Thanks for the PR! 🙏

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-bf9d0a9-20230314 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

fernandojbf pushed a commit to fernandojbf/remix that referenced this pull request Mar 21, 2023
…emix-run#5564)

Co-authored-by: Logan McAnsh <logan@mcan.sh>
Co-authored-by: Mark Dalgleish <mark.john.dalgleish@gmail.com>
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

CSS found in .mjs files is not bundled with unstable_cssSideEffectImports

4 participants