Skip to content

fix: revert browserRouteModulesPlugin - #5721

Merged
jacob-ebey merged 1 commit into
devfrom
client-route-modules
Mar 9, 2023
Merged

jacob-ebey merged 1 commit into
devfrom
client-route-modules

Conversation

@jacob-ebey

@jacob-ebey jacob-ebey commented Mar 8, 2023 •

Copy link
Copy Markdown
Member

fix: put new browserRouteModulesPlugin behind config.future.unstable_dev

This fixes an issue that arose because of using AST parsing to strip route module exports. Reverting to previous way for now and will work on a fix behind the unstable_dev flag.

Closes: #

  • Docs
  • Tests

Testing Strategy:
New integration tests.

@changeset-bot

changeset-bot Bot commented Mar 8, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 225716c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 18 packages
Name Type
@remix-run/dev Minor
@remix-run/react Minor
remix Minor
create-remix Minor
@remix-run/css-bundle Minor
@remix-run/testing Minor
@remix-run/architect Minor
@remix-run/cloudflare Minor
@remix-run/cloudflare-pages Minor
@remix-run/cloudflare-workers Minor
@remix-run/deno Minor
@remix-run/eslint-config Minor
@remix-run/express Minor
@remix-run/netlify Minor
@remix-run/node Minor
@remix-run/serve Minor
@remix-run/server-runtime Minor
@remix-run/vercel Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

fix: put new browserRouteModulesPlugin behind `config.future.unstable_dev`
@jacob-ebey
jacob-ebey force-pushed the client-route-modules branch from a21223c to 225716c Compare March 8, 2023 22:27
@jacob-ebey
jacob-ebey changed the base branch from main to dev March 8, 2023 22:27
@jacob-ebey
jacob-ebey merged commit 91c26ba into dev Mar 9, 2023
@jacob-ebey
jacob-ebey deleted the client-route-modules branch March 9, 2023 00:19
@github-actions

github-actions Bot commented Mar 9, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-375ad78-20230309 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants