Skip to content

fix(remix-dev): handle absolute bundled CSS URLs - #5788

Merged
pcattori merged 4 commits into
devfrom
markdalgleish/absolute-css-urls
Mar 30, 2023
Merged

pcattori merged 4 commits into
devfrom
markdalgleish/absolute-css-urls

Conversation

@markdalgleish

Copy link
Copy Markdown
Member

Closes: #5739

  • Docs
  • Tests

Testing Strategy:

A test case containing an absolute image URL has been added to the test suites for CSS Modules, Vanilla Extract and CSS side-effect imports.

@changeset-bot

changeset-bot Bot commented Mar 14, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 9b58195

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 18 packages
Name Type
@remix-run/dev Patch
create-remix Patch
@remix-run/css-bundle Patch
remix Patch
@remix-run/architect Patch
@remix-run/cloudflare Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/deno Patch
@remix-run/eslint-config Patch
@remix-run/express Patch
@remix-run/netlify Patch
@remix-run/node Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
@remix-run/testing Patch
@remix-run/vercel Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@markdalgleish
markdalgleish requested a review from pcattori March 30, 2023 00:21
@pcattori
pcattori merged commit 9b5cb43 into dev Mar 30, 2023
@pcattori
pcattori deleted the markdalgleish/absolute-css-urls branch March 30, 2023 13:50
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-98dee11-20230331 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

github-actions Bot commented Apr 1, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-27a2acd-20230401 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants