Skip to content

fix(dev): keep dev server running when build fails - #5795

Merged
pcattori merged 2 commits into
release-nextfrom
pedro/fix-dev-server-crash
Mar 15, 2023
Merged

pcattori merged 2 commits into
release-nextfrom
pedro/fix-dev-server-crash

Conversation

@pcattori

@pcattori pcattori commented Mar 14, 2023 •

Copy link
Copy Markdown
Contributor

Closes #5631

@changeset-bot

changeset-bot Bot commented Mar 14, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b1e8270

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 18 packages
Name Type
@remix-run/dev Patch
create-remix Patch
@remix-run/css-bundle Patch
remix Patch
@remix-run/architect Patch
@remix-run/cloudflare Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/deno Patch
@remix-run/eslint-config Patch
@remix-run/express Patch
@remix-run/netlify Patch
@remix-run/node Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
@remix-run/testing Patch
@remix-run/vercel Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

node 19.8 is buggy (nodejs/node#47096)
use 19.7 instead
@pcattori

pcattori commented Mar 15, 2023 •

Copy link
Copy Markdown
Contributor Author

@MichaelDeBoey did you verify that these changes fix #5711 ? I think that issue is related, but not necessarily fixed by this change. I haven't gotten the chance to check that empirically.

The goal of this PR is to fix regressions especially for the old, stable dev server. #5711 doesn't seem to be a regression and is only affecting unstable_dev

@pcattori
pcattori merged commit 6a738d7 into release-next Mar 15, 2023
@pcattori
pcattori deleted the pedro/fix-dev-server-crash branch March 15, 2023 13:52
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.14.3-pre.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.14.3 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

HMR crash when adding a new file 1.14.0 Syntax error exits server

3 participants