Skip to content

chore(remix-dev): use minimatch for regex instead of glob-to-regexp - #6017

Merged
mcansh merged 2 commits into
devfrom
logan/flat-routes-minimatch
Apr 10, 2023
Merged

mcansh merged 2 commits into
devfrom
logan/flat-routes-minimatch

Conversation

@mcansh

@mcansh mcansh commented Apr 7, 2023

Copy link
Copy Markdown
Contributor

Closes: #5983

  • Docs
  • Tests

Testing Strategy:

Signed-off-by: Logan McAnsh <logan@mcan.sh>
@changeset-bot

changeset-bot Bot commented Apr 7, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d7d0cf7

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 18 packages
Name Type
remix Patch
@remix-run/dev Patch
create-remix Patch
@remix-run/css-bundle Patch
@remix-run/architect Patch
@remix-run/cloudflare Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/deno Patch
@remix-run/eslint-config Patch
@remix-run/express Patch
@remix-run/netlify Patch
@remix-run/node Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
@remix-run/testing Patch
@remix-run/vercel Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@mcansh
mcansh marked this pull request as ready for review April 10, 2023 21:14
@mcansh
mcansh merged commit 94d4203 into dev Apr 10, 2023
@mcansh
mcansh deleted the logan/flat-routes-minimatch branch April 10, 2023 21:45
@MichaelDeBoey MichaelDeBoey changed the title chore: use minimatch for regex instead of glob-to-regexp chore(remix-dev): use minimatch for regex instead of glob-to-regexp Apr 10, 2023
@MichaelDeBoey MichaelDeBoey changed the title chore(remix-dev): use minimatch for regex instead of glob-to-regexp chore(remix-dev): use minimatch for regex instead of glob-to-regexp Apr 10, 2023
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-94d4203-20230411 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants