fix(dev): wait until app server is killed before starting a new app server - #6289
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: 0004c51 The changes in this PR will be included in the next version bump. This PR includes changesets to release 18 packages
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
pcattori
force-pushed
the
pedro/fix-dev-server-kill
branch
from
May 3, 2023 02:32
f6609bd to
8b182d3
Compare
pcattori
marked this pull request as ready for review
May 3, 2023 02:52
Contributor
Author
|
Need to see if we can do better than |
pcattori
marked this pull request as draft
May 3, 2023 14:37
pcattori
force-pushed
the
pedro/fix-dev-server-kill
branch
from
May 3, 2023 14:57
5d418e3 to
0004c51
Compare
pcattori
marked this pull request as ready for review
May 3, 2023 14:58
This was referenced May 4, 2023
Contributor
|
🤖 Hello there, We just published version Thanks! |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Fixes #6283
Testing: Added a
setTimeoutlocally to artificially delayp.killto simulate a slow-to-kill process for the app server. Confirmed that this code was resilient and waited for process to be killed before proceeding to spin up new app server instance and thatEADDRINUSEwas not encountered.