Skip to content

fix(dev): kill app server when dev server exits - #6395

Merged
pcattori merged 1 commit into
devfrom
pedro/fix-windows-exit
May 15, 2023
Merged

pcattori merged 1 commit into
devfrom
pedro/fix-windows-exit

Conversation

@pcattori

@pcattori pcattori commented May 15, 2023 •

Copy link
Copy Markdown
Contributor

fixes #6322

@changeset-bot

changeset-bot Bot commented May 15, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 4d2230d

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 18 packages
Name Type
@remix-run/dev Patch
create-remix Patch
@remix-run/css-bundle Patch
remix Patch
@remix-run/architect Patch
@remix-run/cloudflare Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/deno Patch
@remix-run/eslint-config Patch
@remix-run/express Patch
@remix-run/netlify Patch
@remix-run/node Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
@remix-run/testing Patch
@remix-run/vercel Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@pcattori
pcattori force-pushed the pedro/fix-windows-exit branch from ef34bc0 to 4d2230d Compare May 15, 2023 16:48
@pcattori
pcattori merged commit 6cd82a4 into dev May 15, 2023
@pcattori
pcattori deleted the pedro/fix-windows-exit branch May 15, 2023 18:02
@MichaelDeBoey MichaelDeBoey linked an issue May 15, 2023 that may be closed by this pull request
1 task
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-360e77e-20230516 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.16.1-pre.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.16.1 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

"unstable_dev" doesn't kill app server on exit (Windows)

2 participants