Skip to content

fix(dev): better error message when remix-serve is not found - #6477

Merged
pcattori merged 1 commit into
devfrom
pedro/dev-remix-serve-not-found
May 25, 2023
Merged

pcattori merged 1 commit into
devfrom
pedro/dev-remix-serve-not-found

Conversation

@pcattori

@pcattori pcattori commented May 25, 2023 •

Copy link
Copy Markdown
Contributor

Fixes #6268

@changeset-bot

changeset-bot Bot commented May 25, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 68da5e5

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 18 packages
Name Type
@remix-run/dev Patch
create-remix Patch
@remix-run/css-bundle Patch
remix Patch
@remix-run/architect Patch
@remix-run/cloudflare Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/deno Patch
@remix-run/eslint-config Patch
@remix-run/express Patch
@remix-run/netlify Patch
@remix-run/node Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
@remix-run/testing Patch
@remix-run/vercel Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@pcattori
pcattori force-pushed the pedro/dev-remix-serve-not-found branch from 1132d88 to 43e01a5 Compare May 25, 2023 06:01
@pcattori pcattori linked an issue May 25, 2023 that may be closed by this pull request
1 task
@pcattori
pcattori force-pushed the pedro/dev-remix-serve-not-found branch from 43e01a5 to fcbb439 Compare May 25, 2023 14:22
@pcattori
pcattori force-pushed the pedro/dev-remix-serve-not-found branch from fcbb439 to 68da5e5 Compare May 25, 2023 14:24
@pcattori
pcattori merged commit 10bcb92 into dev May 25, 2023
@pcattori
pcattori deleted the pedro/dev-remix-serve-not-found branch May 25, 2023 14:50
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-21504cf-20230526 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

github-actions Bot commented Jun 2, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.17.0-pre.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

github-actions Bot commented Jun 6, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.17.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Remix 1.16.0 new dev server prematurely exits incorrectly

1 participant