Skip to content

fix(dev): cache PostCSS for CSS Modules - #6604

Merged
markdalgleish merged 3 commits into
devfrom
markdalgleish/cache-postcss-css-modules
Jun 15, 2023
Merged

markdalgleish merged 3 commits into
devfrom
markdalgleish/cache-postcss-css-modules

Conversation

@markdalgleish

Copy link
Copy Markdown
Member

This PR hooks the new file watch cache up to PostCSS when processing CSS Modules. This also extends the cache to support arbitrary values, not just strings.

To test this, I've added usage of the composes keyword to the CSS Module in the HMR test so we can validate that dependencies are being tracked.

@changeset-bot

changeset-bot Bot commented Jun 14, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 0070429

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 18 packages
Name Type
@remix-run/dev Patch
create-remix Patch
remix Patch
@remix-run/architect Patch
@remix-run/cloudflare Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/css-bundle Patch
@remix-run/deno Patch
@remix-run/eslint-config Patch
@remix-run/express Patch
@remix-run/netlify Patch
@remix-run/node Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
@remix-run/testing Patch
@remix-run/vercel Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@markdalgleish
markdalgleish merged commit ad9adee into dev Jun 15, 2023
@markdalgleish
markdalgleish deleted the markdalgleish/cache-postcss-css-modules branch June 15, 2023 03:17
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-ad9adee-20230615 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-12440f3-20230616 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.18.0-pre.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.18.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants