Skip to content

perf(dev): cache regular stylesheet compilation - #6638

Merged
pcattori merged 7 commits into
devfrom
markdalgleish/cache-regular-stylesheets
Jun 21, 2023
Merged

pcattori merged 7 commits into
devfrom
markdalgleish/cache-regular-stylesheets

Conversation

@markdalgleish

Copy link
Copy Markdown
Member

This PR uses our new fileWatchCache to cache compilation of regular CSS imports. This avoids the need to run them through an esbuild sub-compilation if the file contents haven't changed.

This change is covered by our existing HMR tests.

@changeset-bot

changeset-bot Bot commented Jun 19, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: abebd77

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 18 packages
Name Type
@remix-run/dev Patch
create-remix Patch
remix Patch
@remix-run/architect Patch
@remix-run/cloudflare Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/css-bundle Patch
@remix-run/deno Patch
@remix-run/eslint-config Patch
@remix-run/express Patch
@remix-run/netlify Patch
@remix-run/node Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
@remix-run/testing Patch
@remix-run/vercel Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@markdalgleish markdalgleish changed the title fix(dev): cache regular stylesheet compilation perf(dev): cache regular stylesheet compilation Jun 21, 2023
@pcattori
pcattori merged commit 1eb1832 into dev Jun 21, 2023
@pcattori
pcattori deleted the markdalgleish/cache-regular-stylesheets branch June 21, 2023 02:51
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-ccefed3-20230621 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.18.0-pre.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.18.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants