Skip to content

fix(dev): do not clear screen when dev server starts - #6719

Merged
pcattori merged 1 commit into
devfrom
pedro/fix-dev-clear
Jun 28, 2023
Merged

pcattori merged 1 commit into
devfrom
pedro/fix-dev-clear

Conversation

@pcattori

@pcattori pcattori commented Jun 28, 2023 •

Copy link
Copy Markdown
Contributor

Fixes #6709

on some terminal emulators, "clearing" only scrolls the next line to the top. on others, it erases the scrollback.

instead, let users call clear themselves (clear && remix dev) if they want to clear

Closes: #

  • Docs
  • Tests

Testing Strategy:

on some terminal emulators, "clearing" only scrolls the next line to the
top. on others, it erases the scrollback.

instead, let users call `clear` themselves (`clear && remix dev`) if
they want to clear
@changeset-bot

changeset-bot Bot commented Jun 28, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 9766aa7

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 18 packages
Name Type
@remix-run/dev Patch
create-remix Patch
remix Patch
@remix-run/architect Patch
@remix-run/cloudflare Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/css-bundle Patch
@remix-run/deno Patch
@remix-run/eslint-config Patch
@remix-run/express Patch
@remix-run/netlify Patch
@remix-run/node Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
@remix-run/testing Patch
@remix-run/vercel Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@pcattori pcattori linked an issue Jun 28, 2023 that may be closed by this pull request
1 task
@pcattori
pcattori merged commit 3775f30 into dev Jun 28, 2023
@pcattori
pcattori deleted the pedro/fix-dev-clear branch June 28, 2023 19:04
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-4ef7a88-20230629 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-0657c16-20230630 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.18.1-pre.2 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

github-actions Bot commented Jul 1, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-7abaf9f-20230701 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.19.0-pre.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.19.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

dev 1.18.0 is clearing my terminal

1 participant