Skip to content

fix(dev): fix sourcemaps for v2_dev - #6762

Merged
pcattori merged 4 commits into
devfrom
pedro/fix-sourcemaps
Jul 5, 2023
Merged

pcattori merged 4 commits into
devfrom
pedro/fix-sourcemaps

Conversation

@pcattori

@pcattori pcattori commented Jul 3, 2023 •

Copy link
Copy Markdown
Contributor

Fixes #6392

Screenshot 2023-07-03 at 4 05 38 PM

TODO

  • changeset
  • Add fences (// REMIX HMR BEGIN, // REMIX HMR END) around the injected HMR code

@changeset-bot

changeset-bot Bot commented Jul 3, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ef11f5e

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 18 packages
Name Type
@remix-run/dev Patch
create-remix Patch
remix Patch
@remix-run/architect Patch
@remix-run/cloudflare Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/css-bundle Patch
@remix-run/deno Patch
@remix-run/eslint-config Patch
@remix-run/express Patch
@remix-run/netlify Patch
@remix-run/node Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
@remix-run/testing Patch
@remix-run/vercel Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@pcattori pcattori linked an issue Jul 3, 2023 that may be closed by this pull request
1 task
@pcattori
pcattori force-pushed the pedro/fix-sourcemaps branch from 47c6541 to 3401b6b Compare July 3, 2023 19:47
@pcattori
pcattori marked this pull request as ready for review July 3, 2023 20:04
@jacob-ebey

Copy link
Copy Markdown
Member

Little worried about doing more work in babel. Mind doing some sort of quick perf verification, maybe against brochure?

@pcattori

pcattori commented Jul 5, 2023 •

Copy link
Copy Markdown
Contributor Author

applyHmr gets run for each route, so testing perf with different # of routes:

App Branch Build Rebuild
Indie stack 1.18.1 400ms 100ms
Indie stack pedro/fix-sourcemaps 400ms 100ms
Indie stack w/ 300 manual routes 1.18.1 2.2s 2.2s
Indie stack w/ 300 manual routes pedro/fix-sourcemaps 2.1s 2.1s

Looks like this PR has same (or even better) perf than 1.18.1

@pcattori
pcattori force-pushed the pedro/fix-sourcemaps branch from 13e85f9 to ef11f5e Compare July 5, 2023 14:57
@pcattori
pcattori merged commit 9762111 into dev Jul 5, 2023
@pcattori
pcattori deleted the pedro/fix-sourcemaps branch July 5, 2023 15:55
@github-actions

github-actions Bot commented Jul 6, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-da44482-20230706 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.19.0-pre.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.19.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

HMR (unstable_dev in 1.16.0) loses TypeScript sourcemap support

2 participants