Skip to content

feat(dev): support empty server polyfills - #6859

Merged
brophdawg11 merged 2 commits into
release-nextfrom
markdalgleish/empty-polyfills
Jul 18, 2023
Merged

brophdawg11 merged 2 commits into
release-nextfrom
markdalgleish/empty-polyfills

Conversation

@markdalgleish

Copy link
Copy Markdown
Member

Closes #6665.

This modifies the API for the serverNodeBuiltinsPolyfill option so that you can also opt into empty polyfills. This is because our old polyfill setup provided empty polyfills by default for fs, crypto, dns, dgram, cluster, repl and tls.

Note that I've nested the modules config within an object (matching the API of esbuild-plugins-node-modules-polyfill) so that we can add more options in the future without it being a breaking change.

@changeset-bot

changeset-bot Bot commented Jul 18, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: f7fa36c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 18 packages
Name Type
@remix-run/dev Minor
create-remix Minor
remix Minor
@remix-run/architect Minor
@remix-run/cloudflare Minor
@remix-run/cloudflare-pages Minor
@remix-run/cloudflare-workers Minor
@remix-run/css-bundle Minor
@remix-run/deno Minor
@remix-run/eslint-config Minor
@remix-run/express Minor
@remix-run/netlify Minor
@remix-run/node Minor
@remix-run/react Minor
@remix-run/serve Minor
@remix-run/server-runtime Minor
@remix-run/testing Minor
@remix-run/vercel Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@brophdawg11
brophdawg11 merged commit 472c936 into release-next Jul 18, 2023
@brophdawg11
brophdawg11 deleted the markdalgleish/empty-polyfills branch July 18, 2023 12:07
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.19.0-pre.4 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.19.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants