Skip to content

feat(remix-dev)!: remove serverBuildTarget config option - #6896

Merged
brophdawg11 merged 2 commits into
remix-run:devfrom
MichaelDeBoey:remove-serverBuildTarget-config
Jul 20, 2023
Merged

brophdawg11 merged 2 commits into
remix-run:devfrom
MichaelDeBoey:remove-serverBuildTarget-config

Conversation

@MichaelDeBoey

Copy link
Copy Markdown
Member

Resubmission of #5635

@changeset-bot

changeset-bot Bot commented Jul 20, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: ffeadc0

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 18 packages
Name Type
@remix-run/dev Major
create-remix Major
remix Major
@remix-run/architect Major
@remix-run/cloudflare Major
@remix-run/cloudflare-pages Major
@remix-run/cloudflare-workers Major
@remix-run/css-bundle Major
@remix-run/deno Major
@remix-run/eslint-config Major
@remix-run/express Major
@remix-run/netlify Major
@remix-run/node Major
@remix-run/react Major
@remix-run/serve Major
@remix-run/server-runtime Major
@remix-run/testing Major
@remix-run/vercel Major

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@MichaelDeBoey
MichaelDeBoey force-pushed the remove-serverBuildTarget-config branch from 152d8fc to a10d371 Compare July 20, 2023 18:22
@brophdawg11
brophdawg11 merged commit 4614557 into remix-run:dev Jul 20, 2023
@MichaelDeBoey
MichaelDeBoey deleted the remove-serverBuildTarget-config branch July 20, 2023 19:39
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-657c9d5-20230721 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-974954f-20230722 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

github-actions Bot commented Aug 2, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-3b808ce-20230802 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

github-actions Bot commented Aug 2, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.19.2-pre.1 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

github-actions Bot commented Aug 3, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-7cb1e7e-20230803 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

github-actions Bot commented Aug 4, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.19.2-pre.2 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

github-actions Bot commented Aug 5, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-6e250f7-20230805 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-b1149bb-20230810 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

No open projects
Archived in project

Development

Successfully merging this pull request may close these issues.

2 participants