Skip to content

feat(remix-dev)!: remove devServerBroadcastDelay config option - #7063

Merged
MichaelDeBoey merged 1 commit into
remix-run:devfrom
MichaelDeBoey:remove-devServerBroadcastDelay_config-option
Aug 14, 2023
Merged

MichaelDeBoey merged 1 commit into
remix-run:devfrom
MichaelDeBoey:remove-devServerBroadcastDelay_config-option

Conversation

@MichaelDeBoey

Copy link
Copy Markdown
Member

No description provided.

@changeset-bot

changeset-bot Bot commented Aug 4, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b6793ed

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 16 packages
Name Type
@remix-run/dev Major
create-remix Major
remix Major
@remix-run/architect Major
@remix-run/cloudflare Major
@remix-run/cloudflare-pages Major
@remix-run/cloudflare-workers Major
@remix-run/css-bundle Major
@remix-run/deno Major
@remix-run/eslint-config Major
@remix-run/express Major
@remix-run/node Major
@remix-run/react Major
@remix-run/serve Major
@remix-run/server-runtime Major
@remix-run/testing Major

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@brophdawg11
brophdawg11 removed their request for review August 4, 2023 14:39
@MichaelDeBoey
MichaelDeBoey force-pushed the remove-devServerBroadcastDelay_config-option branch 6 times, most recently from ab6437c to da5db7d Compare August 8, 2023 22:06
Comment thread packages/remix-dev/devServer/liveReload.ts Outdated
@MichaelDeBoey
MichaelDeBoey force-pushed the remove-devServerBroadcastDelay_config-option branch 6 times, most recently from 18ce73c to 9ba8901 Compare August 11, 2023 17:56
@MichaelDeBoey
MichaelDeBoey force-pushed the remove-devServerBroadcastDelay_config-option branch from 9ba8901 to c8e7b85 Compare August 14, 2023 13:48
@brophdawg11
brophdawg11 removed their request for review August 14, 2023 16:16
@MichaelDeBoey
MichaelDeBoey force-pushed the remove-devServerBroadcastDelay_config-option branch 2 times, most recently from bc7860e to e82d6f9 Compare August 14, 2023 18:14
@MichaelDeBoey
MichaelDeBoey force-pushed the remove-devServerBroadcastDelay_config-option branch from e82d6f9 to b6793ed Compare August 14, 2023 19:20
@MichaelDeBoey
MichaelDeBoey merged commit 7a9ed0b into remix-run:dev Aug 14, 2023
@MichaelDeBoey
MichaelDeBoey deleted the remove-devServerBroadcastDelay_config-option branch August 14, 2023 19:41
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-65fdbcd-20230815 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

No open projects
Status: Merged

Development

Successfully merging this pull request may close these issues.

3 participants