Skip to content

fix(remix-dev): do not interpret JSX in .ts files - #7306

Merged
pcattori merged 1 commit into
release-nextfrom
pedro/fix-ts-vs-tsx
Aug 30, 2023
Merged

pcattori merged 1 commit into
release-nextfrom
pedro/fix-ts-vs-tsx

Conversation

@pcattori

@pcattori pcattori commented Aug 30, 2023 •

Copy link
Copy Markdown
Contributor

Fixes #7295

@pcattori pcattori linked an issue Aug 30, 2023 that may be closed by this pull request
1 task
@pcattori
pcattori merged commit 7300cf8 into release-next Aug 30, 2023
@pcattori
pcattori deleted the pedro/fix-ts-vs-tsx branch August 30, 2023 19:58
@changeset-bot

changeset-bot Bot commented Aug 30, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 3ae11f6

The changes in this PR will be included in the next version bump.

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@MichaelDeBoey MichaelDeBoey changed the title do not interpret JSX in .ts files fix(remix-dev): do not interpret JSX in .ts files Aug 30, 2023
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.0.0-pre.4 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@xHomu

xHomu commented Aug 31, 2023

Copy link
Copy Markdown
Contributor

Can confirm 2.0.0-pre.4 resolved this
image

@github-actions

github-actions Bot commented Sep 1, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-6c0310c-20230901 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

github-actions Bot commented Sep 1, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.0.0-pre.6 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

github-actions Bot commented Sep 2, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-9288516-20230902 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

github-actions Bot commented Sep 5, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.0.0-pre.7 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

1 similar comment
@github-actions

github-actions Bot commented Sep 5, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.0.0-pre.7 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

github-actions Bot commented Sep 6, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-481f73e-20230906 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

github-actions Bot commented Sep 6, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.0.0-pre.8 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

github-actions Bot commented Sep 7, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-1a57073-20230907 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

github-actions Bot commented Sep 7, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.0.0-pre.9 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@xHomu

xHomu commented Sep 7, 2023

Copy link
Copy Markdown
Contributor

Can confirm 2.0.0-pre.9 working in production ✅ https://github.com/manawiki/core/commit/d15667066cf2ed0c597bf078968e02d0b70fb638

Is v2 is almost out of the oven?

@github-actions

github-actions Bot commented Sep 8, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-1fac238-20230908 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.0.0-pre.10 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-3646f91-20230914 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

No open projects
Status: Merged

Development

Successfully merging this pull request may close these issues.

v2pre ts bug, not happy with typed const arrow functions

3 participants