Skip to content

feat: add unstable Vite support - #7590

Merged
pcattori merged 39 commits into
devfrom
experimental-vite
Oct 13, 2023
Merged

pcattori merged 39 commits into
devfrom
experimental-vite

Conversation

@markdalgleish

@markdalgleish markdalgleish commented Oct 5, 2023 •

Copy link
Copy Markdown
Member

This PR is based on the initial Vite prototype in pcattori/revive built by @pcattori and myself.

TODO:

  • Finish initial docs
    • Move to new Future section
    • Hosting/runtime/feature compatibility matrix
    • Call out HMR differences
    • Thank everyone who provided prior art and/or inspired our work
  • Finalise API
    • Rename "experimental" to "unstable"
    • Move plugin from @remix-run/dev/vite to @remix-run/dev
    • Remove Remark plugin (for now at least)
  • Dedupe config resolution logic being used inside Vite plugin
  • Warn about unstable usage in terminal

@changeset-bot

changeset-bot Bot commented Oct 5, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: b605be3

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 17 packages
Name Type
integration-tests Minor
create-remix Minor
@remix-run/dev Minor
@remix-run/react Minor
@remix-run/server-runtime Minor
@remix-run/testing Minor
@remix-run/cloudflare Minor
@remix-run/deno Minor
@remix-run/node Minor
@remix-run/cloudflare-pages Minor
@remix-run/cloudflare-workers Minor
@remix-run/architect Minor
@remix-run/express Minor
@remix-run/serve Minor
remix Minor
@remix-run/css-bundle Minor
@remix-run/eslint-config Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@markdalgleish
markdalgleish marked this pull request as ready for review October 13, 2023 03:26
@markdalgleish markdalgleish changed the title [WIP] Add experimental Vite support feat: add unstable Vite support Oct 13, 2023
Comment thread docs/future/vite.md Outdated
Co-authored-by: Pedro Cattori <pcattori@gmail.com>
Comment thread docs/future/vite.md Outdated
Co-authored-by: Pedro Cattori <pcattori@gmail.com>
Comment thread docs/future/vite.md Outdated
Comment thread docs/future/vite.md Outdated
Comment thread docs/future/vite.md Outdated
Comment thread docs/future/vite.md
@pcattori
pcattori merged commit 1ceae7b into dev Oct 13, 2023
@pcattori
pcattori deleted the experimental-vite branch October 13, 2023 16:33
@remix-run remix-run deleted a comment from github-actions Bot Oct 16, 2023
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.2.0-pre.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@mmachatschek

mmachatschek commented Oct 30, 2023 •

Copy link
Copy Markdown

@pcattori where should issues be reported?

Edit: I will create a normal issue report with the mention about the nightly (even though maybe a discussion whould be better to gather feedback about the experimental feature?)

@pcattori

Copy link
Copy Markdown
Contributor

Yea separate issue with "Vite Nightly" or something like that is the way to go

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.2.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants