Skip to content

fix(vite): fix dev FOUC with custom Express server - #7937

Merged
markdalgleish merged 1 commit into
devfrom
markdalgleish/vite-dev-express-fouc
Nov 8, 2023
Merged

markdalgleish merged 1 commit into
devfrom
markdalgleish/vite-dev-express-fouc

Conversation

@markdalgleish

Copy link
Copy Markdown
Member

Fixes #7878.

@changeset-bot

changeset-bot Bot commented Nov 8, 2023

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 1ed50d4

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 16 packages
Name Type
@remix-run/dev Patch
@remix-run/express Patch
@remix-run/serve Patch
create-remix Patch
remix Patch
@remix-run/architect Patch
@remix-run/cloudflare Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/css-bundle Patch
@remix-run/deno Patch
@remix-run/eslint-config Patch
@remix-run/node Patch
@remix-run/react Patch
@remix-run/server-runtime Patch
@remix-run/testing Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@markdalgleish
markdalgleish merged commit 09d557a into dev Nov 8, 2023
@markdalgleish
markdalgleish deleted the markdalgleish/vite-dev-express-fouc branch November 8, 2023 22:36
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.3.0-pre.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@gijsroge

Copy link
Copy Markdown

@markdalgleish just tested the pre-version, works flawlessly.

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.3.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants