Skip to content

feat(remix-dev/vite): Add unstable_serverBundles - #8332

Merged
markdalgleish merged 19 commits into
devfrom
markdalgleish/server-bundles
Dec 21, 2023
Merged

markdalgleish merged 19 commits into
devfrom
markdalgleish/server-bundles

Conversation

@markdalgleish

@markdalgleish markdalgleish commented Dec 19, 2023 •

Copy link
Copy Markdown
Member

See updated docs for details, but the high level summary is that this lets you split your server into multiple bundles with a request handler in each, with each one designed to handle a subset of routes.

We also generate a manifest in build/server/bundles.json so consumers have everything they need to direct requests to the correct server bundle. We may provide higher level utilities for consuming the manifest, but this is good enough to get the ball rolling.

Also see #8296.

@changeset-bot

changeset-bot Bot commented Dec 19, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: a1a41b5

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 16 packages
Name Type
@remix-run/dev Minor
create-remix Minor
remix Minor
@remix-run/architect Minor
@remix-run/cloudflare Minor
@remix-run/cloudflare-pages Minor
@remix-run/cloudflare-workers Minor
@remix-run/css-bundle Minor
@remix-run/deno Minor
@remix-run/eslint-config Minor
@remix-run/express Minor
@remix-run/node Minor
@remix-run/react Minor
@remix-run/serve Minor
@remix-run/server-runtime Minor
@remix-run/testing Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

Comment thread docs/future/server-bundles.md Outdated
Comment thread docs/future/server-bundles.md Outdated
Comment thread docs/future/server-bundles.md Outdated
Comment thread docs/future/server-bundles.md Outdated
Comment thread docs/future/server-bundles.md Outdated
Comment thread integration/vite-server-bundles-test.ts Outdated
Comment thread packages/remix-dev/vite/build.ts Outdated
@markdalgleish
markdalgleish marked this pull request as ready for review December 19, 2023 23:17
@markdalgleish
markdalgleish dismissed pcattori’s stale review December 20, 2023 00:29

Addressed requested changes

@markdalgleish
markdalgleish merged commit e9440e6 into dev Dec 21, 2023
@markdalgleish
markdalgleish deleted the markdalgleish/server-bundles branch December 21, 2023 23:20
@github-actions

github-actions Bot commented Jan 9, 2024

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.5.0-pre.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.5.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants