Skip to content

fix(remix-dev/vite): default NODE_ENV in build - #8405

Merged
markdalgleish merged 1 commit into
devfrom
markdalgleish/set-node-env-in-vite-build
Jan 2, 2024
Merged

markdalgleish merged 1 commit into
devfrom
markdalgleish/set-node-env-in-vite-build

Conversation

@markdalgleish

Copy link
Copy Markdown
Member

Fixes #8364.

This PR implements the fix suggested here: #8365 (comment)

I've also added a test and ensured that it fails without the provided fix.

@changeset-bot

changeset-bot Bot commented Jan 1, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 4f5504a

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 16 packages
Name Type
@remix-run/dev Patch
create-remix Patch
remix Patch
@remix-run/architect Patch
@remix-run/cloudflare Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/css-bundle Patch
@remix-run/deno Patch
@remix-run/eslint-config Patch
@remix-run/express Patch
@remix-run/node Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
@remix-run/testing Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@markdalgleish
markdalgleish merged commit 81c7a99 into dev Jan 2, 2024
@markdalgleish
markdalgleish deleted the markdalgleish/set-node-env-in-vite-build branch January 2, 2024 00:09
@github-actions

github-actions Bot commented Jan 9, 2024

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.5.0-pre.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.5.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant