Skip to content

fix(remix-run/vite): loosen route export restrictions - #8420

Merged
markdalgleish merged 5 commits into
devfrom
pedro/no-strict-route-exports
Jan 9, 2024
Merged

markdalgleish merged 5 commits into
devfrom
pedro/no-strict-route-exports

Conversation

@pcattori

@pcattori pcattori commented Jan 3, 2024 •

Copy link
Copy Markdown
Contributor

TODO

  • Document that route exports are not removed from client bundle in dev

@changeset-bot

changeset-bot Bot commented Jan 3, 2024 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 9e42efe

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 16 packages
Name Type
@remix-run/dev Patch
create-remix Patch
remix Patch
@remix-run/architect Patch
@remix-run/cloudflare Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/css-bundle Patch
@remix-run/deno Patch
@remix-run/eslint-config Patch
@remix-run/express Patch
@remix-run/node Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
@remix-run/testing Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@markdalgleish markdalgleish changed the title vite: loosen route export restrictions fix(remix-run/vite): loosen route export restrictions Jan 9, 2024
@markdalgleish
markdalgleish merged commit 4f0a730 into dev Jan 9, 2024
@markdalgleish
markdalgleish deleted the pedro/no-strict-route-exports branch January 9, 2024 03:56
@github-actions

github-actions Bot commented Jan 9, 2024

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.5.0-pre.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.5.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants