Skip to content

vite: profiling - #8493

Merged
pcattori merged 5 commits into
devfrom
pedro/vite-profile
Jan 12, 2024
Merged

pcattori merged 5 commits into
devfrom
pedro/vite-profile

Conversation

@pcattori

@pcattori pcattori commented Jan 11, 2024 •

Copy link
Copy Markdown
Contributor

TODO

  • Docs
  • nicer changeset

Creates .cpuprofiles that can be uploaded to speedscope.app

Screenshot 2024-01-11 at 4 04 39 PM

@changeset-bot

changeset-bot Bot commented Jan 11, 2024 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: a6245ba

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 16 packages
Name Type
@remix-run/dev Patch
create-remix Patch
remix Patch
@remix-run/architect Patch
@remix-run/cloudflare Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/css-bundle Patch
@remix-run/deno Patch
@remix-run/eslint-config Patch
@remix-run/express Patch
@remix-run/node Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
@remix-run/testing Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@pcattori
pcattori changed the base branch from main to dev January 11, 2024 21:05
Comment thread packages/remix-dev/vite/profiler.ts Outdated
Comment thread packages/remix-dev/vite/profiler.ts Outdated
Comment thread packages/remix-dev/vite/profiler.ts Outdated
build: `--profile`
dev:
  `--profile` to start with profiler
  `p + enter` to write profile and restart profiler
Comment thread packages/remix-dev/cli/commands.ts
@pcattori
pcattori merged commit 4b7fb14 into dev Jan 12, 2024
@pcattori
pcattori deleted the pedro/vite-profile branch January 12, 2024 17:32
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.5.1-pre.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.5.1 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants