Skip to content

vite: add sourcemap flags - #8613

Merged
pcattori merged 4 commits into
devfrom
pedro/vite-sourcemap-flag
Jan 30, 2024
Merged

pcattori merged 4 commits into
devfrom
pedro/vite-sourcemap-flag

Conversation

@pcattori

@pcattori pcattori commented Jan 28, 2024 •

Copy link
Copy Markdown
Contributor
Screenshot 2024-01-30 at 1 07 55 AM Screenshot 2024-01-30 at 1 08 22 AM

@changeset-bot

changeset-bot Bot commented Jan 28, 2024 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d179279

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 16 packages
Name Type
@remix-run/dev Patch
create-remix Patch
remix Patch
@remix-run/architect Patch
@remix-run/cloudflare Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/css-bundle Patch
@remix-run/deno Patch
@remix-run/eslint-config Patch
@remix-run/express Patch
@remix-run/node Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
@remix-run/testing Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@mmachatschek

Copy link
Copy Markdown

Is there any chance that the --sourcemap flag could be narrowed down to server/client side only generation?

#5202

E.g. --sourcemap enables sourcemap generation for both server and client, --sourcemap-server for server side only code and --sourcemap-client for client side only code.

I suppose this is easier to solve with the vite setup than with the old compiler.

Comment thread .changeset/curly-camels-applaud.md Outdated
@pcattori
pcattori force-pushed the pedro/vite-sourcemap-flag branch from 104658d to d179279 Compare January 30, 2024 06:06
@pcattori pcattori changed the title vite: add --sourcemap flag vite: add sourcemap flags Jan 30, 2024
@pcattori
pcattori marked this pull request as ready for review January 30, 2024 06:42
@pcattori
pcattori merged commit c372825 into dev Jan 30, 2024
@pcattori
pcattori deleted the pedro/vite-sourcemap-flag branch January 30, 2024 06:47
@kentcdodds

Copy link
Copy Markdown
Member

Ha! This explains why my Sentry reports were suddenly much harder to read 😅 I totally forgot about sourcemaps!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.6.0-pre.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

github-actions Bot commented Feb 1, 2024

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.6.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants