Skip to content

Stabilize SPA Mode - #8692

Merged
brophdawg11 merged 8 commits into
devfrom
brophdawg11/stabilize-spa-mode
Feb 8, 2024
Merged

brophdawg11 merged 8 commits into
devfrom
brophdawg11/stabilize-spa-mode

Conversation

@brophdawg11

@brophdawg11 brophdawg11 commented Feb 6, 2024 •

Copy link
Copy Markdown
Contributor

Todo:

  • Remove unstable note in docs
  • Add some docs on how to hydration only a div instead of the whole document
  • Add a note to docs about the SSR pass and how to solve it using React.lazy or ClientOnly
  • Re-export Routes/Route up through @remix-run/react for migrations within a splat route

@changeset-bot

changeset-bot Bot commented Feb 6, 2024 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 39c0ccf

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 16 packages
Name Type
@remix-run/dev Minor
create-remix Minor
remix Minor
@remix-run/architect Minor
@remix-run/cloudflare Minor
@remix-run/cloudflare-pages Minor
@remix-run/cloudflare-workers Minor
@remix-run/css-bundle Minor
@remix-run/deno Minor
@remix-run/eslint-config Minor
@remix-run/express Minor
@remix-run/node Minor
@remix-run/react Minor
@remix-run/serve Minor
@remix-run/server-runtime Minor
@remix-run/testing Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@brophdawg11
brophdawg11 merged commit f3877f5 into dev Feb 8, 2024
@brophdawg11
brophdawg11 deleted the brophdawg11/stabilize-spa-mode branch February 8, 2024 20:49
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.7.0-pre.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.7.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants