Skip to content

vite: fix server export dead-code elimination for routes outside of app dir - #8795

Merged
pcattori merged 1 commit into
release-nextfrom
pedro/vite-fix-routes-outside-app-dir
Feb 19, 2024
Merged

pcattori merged 1 commit into
release-nextfrom
pedro/vite-fix-routes-outside-app-dir

Conversation

@pcattori

@pcattori pcattori commented Feb 19, 2024 •

Copy link
Copy Markdown
Contributor

Fixes #8790 #6711

Co-authored-by: Luciano Fantone <luciano.alvarezfantone@sas.se>
@changeset-bot

changeset-bot Bot commented Feb 19, 2024

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7a98d9f

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 16 packages
Name Type
@remix-run/dev Patch
create-remix Patch
remix Patch
@remix-run/architect Patch
@remix-run/cloudflare Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/css-bundle Patch
@remix-run/deno Patch
@remix-run/eslint-config Patch
@remix-run/express Patch
@remix-run/node Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
@remix-run/testing Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@pcattori pcattori changed the title vite: fix server exports elimination for routes outside of app dir vite: fix server export dead-code elimination for routes outside of app dir Feb 19, 2024
@pcattori
pcattori merged commit 9c45259 into release-next Feb 19, 2024
@pcattori
pcattori deleted the pedro/vite-fix-routes-outside-app-dir branch February 19, 2024 21:32
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.7.0-pre.2 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.7.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant