Skip to content

Always ignore route files starting with . - #8801

Merged
pcattori merged 3 commits into
release-nextfrom
markdalgleish/ignore-route-dotfiles
Feb 20, 2024
Merged

pcattori merged 3 commits into
release-nextfrom
markdalgleish/ignore-route-dotfiles

Conversation

@markdalgleish

Copy link
Copy Markdown
Member

This change is in response to the following issue: #8744

Having dotfiles like .DS_Store in the routes directory causes hard-to-debug errors when "**/.*" is not configured in ignoredRouteFiles. This is particularly notable when using the Vite plugin since we've encourage usage without any config (i.e. remix() in the plugins array). Currently none of the Vite templates have ignoredRouteFiles set.

Since we've moved to flat routes, it no longer makes sense to have route files starting with a . character since it's meant to be a delimiter denoting a / in the URL. As a result, rather than doubling down and adding remix({ ignoredRouteFiles: ["**/.*"] }) to all Vite templates, it makes more sense to always ignore route files starting with ..

The bulk of this PR is just updating docs and templates now that it's redundant to add **/.* to ignoredRouteFiles.

@changeset-bot

changeset-bot Bot commented Feb 20, 2024 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 74cda7d

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 16 packages
Name Type
@remix-run/dev Patch
create-remix Patch
remix Patch
@remix-run/architect Patch
@remix-run/cloudflare Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/css-bundle Patch
@remix-run/deno Patch
@remix-run/eslint-config Patch
@remix-run/express Patch
@remix-run/node Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
@remix-run/testing Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@pcattori
pcattori marked this pull request as draft February 20, 2024 17:15
@pcattori
pcattori marked this pull request as ready for review February 20, 2024 17:16
@pcattori
pcattori force-pushed the markdalgleish/ignore-route-dotfiles branch 5 times, most recently from 0b8bbac to 4df8b3d Compare February 20, 2024 18:49
@pcattori
pcattori force-pushed the markdalgleish/ignore-route-dotfiles branch from 096d8e8 to 74cda7d Compare February 20, 2024 19:31
@pcattori
pcattori merged commit c1a32c2 into release-next Feb 20, 2024
@pcattori
pcattori deleted the markdalgleish/ignore-route-dotfiles branch February 20, 2024 20:07
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.7.0-pre.4 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.7.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants