Skip to content

fix: mark Layout as browser safe route export - #8842

Merged
jacob-ebey merged 2 commits into
devfrom
legacy-compiler-layout
Feb 21, 2024
Merged

jacob-ebey merged 2 commits into
devfrom
legacy-compiler-layout

Conversation

@jacob-ebey

Copy link
Copy Markdown
Member

Closes: #

  • Docs
  • Tests

Testing Strategy:

@changeset-bot

changeset-bot Bot commented Feb 21, 2024 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: d4c2912

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 16 packages
Name Type
remix Patch
@remix-run/dev Patch
create-remix Patch
@remix-run/architect Patch
@remix-run/cloudflare Patch
@remix-run/cloudflare-pages Patch
@remix-run/cloudflare-workers Patch
@remix-run/css-bundle Patch
@remix-run/deno Patch
@remix-run/eslint-config Patch
@remix-run/express Patch
@remix-run/node Patch
@remix-run/react Patch
@remix-run/serve Patch
@remix-run/server-runtime Patch
@remix-run/testing Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@jacob-ebey jacob-ebey changed the title fix: mark Layout as browser save route export fix: mark Layout as browser safe route export Feb 21, 2024
@jacob-ebey
jacob-ebey merged commit dc5cdb7 into dev Feb 21, 2024
@jacob-ebey
jacob-ebey deleted the legacy-compiler-layout branch February 21, 2024 18:43
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.8.0-pre.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.8.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants