Skip to content

Pass resolved viteConfig to buildEnd hook - #8885

Merged
markdalgleish merged 3 commits into
devfrom
markdalgleish/buildend-viteconfig
Feb 26, 2024
Merged

markdalgleish merged 3 commits into
devfrom
markdalgleish/buildend-viteconfig

Conversation

@markdalgleish

@markdalgleish markdalgleish commented Feb 26, 2024 •

Copy link
Copy Markdown
Member

This lets presets have access to the resolved Vite config after the build is complete.

This is primarily being driven by the in-progress Vercel preset so that they know which static assets can be cached indefinitely, especially now that we've fixed support for custom asset directories in #8843.

It's worth calling out that presets only get access to the Vite config once the build is complete, so this doesn't open a back door for them to modify the config.

cc @TooTallNate.

@changeset-bot

changeset-bot Bot commented Feb 26, 2024 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: bc615c7

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 16 packages
Name Type
@remix-run/dev Minor
create-remix Minor
remix Minor
@remix-run/architect Minor
@remix-run/cloudflare Minor
@remix-run/cloudflare-pages Minor
@remix-run/cloudflare-workers Minor
@remix-run/css-bundle Minor
@remix-run/deno Minor
@remix-run/eslint-config Minor
@remix-run/express Minor
@remix-run/node Minor
@remix-run/react Minor
@remix-run/serve Minor
@remix-run/server-runtime Minor
@remix-run/testing Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@markdalgleish
markdalgleish merged commit 01cac7b into dev Feb 26, 2024
@markdalgleish
markdalgleish deleted the markdalgleish/buildend-viteconfig branch February 26, 2024 21:01
TooTallNate pushed a commit to vercel/remix that referenced this pull request Feb 26, 2024
@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.8.0-pre.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 2.8.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant