Skip to content

fix(dev): better networking options - #6724

Merged
pcattori merged 5 commits into
devfrom
pedro/fix-dev-networking
Jul 3, 2023
Merged

pcattori merged 5 commits into
devfrom
pedro/fix-dev-networking

Conversation

@pcattori

@pcattori pcattori commented Jun 29, 2023 •

Copy link
Copy Markdown
Contributor

Fixes #6718
Obsoletes #5467

Tested my running creating TLS key/cert for myhost and running:

REMIX_DEV_ORIGIN=https://myhost remix dev --no-restart -c \"node server.js\" --tls-key=key.pem --tls-cert=cert.pem --port=443

👆 Note that origin can omit the port since the default HTTPS port of 443 is used. Also tested with tlsKey and tlsCert config options.

Also tried with the same command without REMIX_DEV_ORIGIN to ensure that TLS protocols (https/wss) are correctly inferred from TLS flags and config options.

Finally, also tested that REMIX_DEV_SERVER_WS_PORT works to set the ws port for the old dev server.

TODO

  • docs
  • changeset

@changeset-bot

changeset-bot Bot commented Jun 29, 2023 •

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 9511fb9

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 18 packages
Name Type
@remix-run/dev Minor
@remix-run/react Minor
@remix-run/server-runtime Minor
create-remix Minor
@remix-run/testing Minor
@remix-run/cloudflare Minor
@remix-run/deno Minor
@remix-run/node Minor
@remix-run/cloudflare-pages Minor
@remix-run/cloudflare-workers Minor
@remix-run/architect Minor
@remix-run/express Minor
@remix-run/netlify Minor
@remix-run/serve Minor
@remix-run/vercel Minor
remix Minor
@remix-run/css-bundle Minor
@remix-run/eslint-config Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@pcattori
pcattori force-pushed the pedro/fix-dev-networking branch from 486d74f to debfbb9 Compare June 29, 2023 16:21
@pcattori pcattori linked an issue Jun 29, 2023 that may be closed by this pull request
1 task
@pcattori
pcattori force-pushed the pedro/fix-dev-networking branch 2 times, most recently from e9d1bd4 to 3273c23 Compare June 29, 2023 20:58
@pcattori
pcattori requested review from blittle and frandiox June 29, 2023 23:28
@pcattori
pcattori force-pushed the pedro/fix-dev-networking branch 2 times, most recently from 4d589b0 to 12b2eb7 Compare June 30, 2023 20:10
@github-actions

github-actions Bot commented Jul 4, 2023

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version v0.0.0-nightly-5ad8fa5-20230704 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.19.0-pre.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

@github-actions

Copy link
Copy Markdown
Contributor

🤖 Hello there,

We just published version 1.19.0 which includes this pull request. If you'd like to take it for a test run please try it out and let us know what you think!

Thanks!

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Scheme not set to HTTPS when TLS config options are set in remix.config.js

1 participant